Skip to content

Bumped development release number for 1.35 - #4366

Closed
Ganiredi wants to merge 1 commit into
aws:mainfrom
Ganiredi:1.35-1_development_1765321096
Closed

Ganiredi wants to merge 1 commit into
aws:mainfrom
Ganiredi:1.35-1_development_1765321096

Conversation

@Ganiredi

@Ganiredi Ganiredi commented Dec 9, 2025

Copy link
Copy Markdown
Contributor

By submitting this pull request, I confirm that you can use, modify, copy, and redistribute this contribution, under the terms of your choice.

@eks-distro-bot

Copy link
Copy Markdown
Collaborator

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by:
Once this PR has been reviewed and has the lgtm label, please ask for approval from ganiredi. For more information see the Kubernetes Code Review Process.

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants